已合并
适配新CANN包下python env check #608
适配新CANN包下python env check #608
已合并
xiongpan创建于 1月26日
xiongpan成员
1月26日

Memory allocator hook registration

What type of PR is this?

/kind bug


What does this PR do / why do we need it:


Which issue(s) this PR fixes:

Fixes #207


Test Plan and Test result:What scenarios were tested, and what were the verification results(Function, performance, reliability, etc.):
新增ascend env checker中对ascend version file校验


Self-checklist:(请自检,在[ ]内打上x,我们将检视你的完成情况,否则会导致pr无法合入)

likedislike
Pull Request已成功合入, 合并人@MindSpore-Bot
(感谢 xiongpan 的贡献)
Xxiongpan成员
1月26日 创建了 pull request,commit 015f8028
MindSpore-BotMindSpore-Bot成员
1月26日 添加了label:mindspore-cla/yes
MindSpore-BotMindSpore-Bot成员
1月26日 添加了label:no-pass-all-review
MindSpore-Bot
MindSpore-Bot成员
1月26日 评论:

Notice

The PR needs 2 assignees to review. 2 does not review. if all are passed, please comment /check-pr to try merge the PR. 😄

likedislike
xiongpan成员
1月26日 评论:

/retest

likedislike
此处折叠了45条消息 查看更多
MindSpore-Bot
MindSpore-Bot成员
2月5日 评论:

Notice

The PR needs 2 assignees to review. 1 does not review. if all are passed, please comment /check-pr to try merge the PR. 😄

likedislike
Jjjfeing成员
2月5日 通过审查
MindSpore-BotMindSpore-Bot成员
2月5日 删除了label:no-pass-all-review
MindSpore-BotMindSpore-Bot成员
2月5日 合入了pull request,合并节点 SHA:9cda60f07ba18fb921acbceba1716e83c97b783a
Lliuf9成员
2月26日 关联了issue:[Bug]: 基于官网mindspore_lite 2.8.0 python API包执行FireRedASR的mindir模型推理时报错no attribute 'ascend_version_file'