已合并
fix updateweight bug #789
lilinjie11创建于 4月16日
fix updateweight bug #789
已合并
Pull Request已成功合入, 合并人@MindSpore-Bot
(感谢 lilinjie11 的贡献)4月16日 创建了 pull request,commit 43d88f71
Llilinjie11
4月16日 关联了issue:[Bug]: linux_aarch64_300iduo环境chs_4k_dcg_ares_20241023_3200_12800_fix.onnx模型推理更新权重报错Run update weights failed! ret:500002!
4月16日 关联了issue:[Bug]: linux_aarch64_300iduo环境chs_4k_dcg_ares_20241023_3200_12800_fix.onnx模型推理更新权重报错Run update weights failed! ret:500002!
4月16日 添加了label:mindspore-cla/yes
4月16日 添加了label:no-pass-all-review
MindSpore-Bot
4月16日 评论:
4月16日 评论:
lilinjie11
4月16日 评论:
4月16日 评论:
/retest


4月16日 预合并成功(commit_id: c0e7af93393d5a1ad152d0bb774c8976f0d341bb)
4月16日 审查状态已重置,审查人: YeFeng_24,jjfeing
4月16日 强制推送 1 个提交:3d396c0b-fix updateweight bug
4月16日 预合并成功(commit_id: e04fa9890d9e43452168f3992ca71b7e6a78a980)
4月16日 添加了label:ci-pipeline-running
4月16日 预合并成功(commit_id: ede458a81edc49b24737d25ca4d98308d8db4dd3)
4月16日 审查状态已重置,审查人: YeFeng_24,jjfeing
4月16日 强制推送 1 个提交:8fb74a7f-fix updateweight bug
4月16日 预合并成功(commit_id: 8c60368c006362ceec5e63f1e482af647c7939e2)
MindSpore-Bot
4月16日 评论:
4月16日 评论:
lilinjie11
4月16日 评论:
4月16日 评论:
/retest


司小南(机器人)
4月16日 评论:
4月16日 评论:
| Project Name | Build_Stage | Build Result | Details |
|---|---|---|---|
| MindSpore-Lite_Atomgit_Gate | - | ❌ FAILURE | 1151 |


4月16日 删除了label:ci-pipeline-running
4月16日 删除了label:ci-pipeline-failed
4月16日 添加了label:ci-pipeline-running
司小南(机器人)
4月16日 评论:
4月16日 评论:
| Project Name | Build_Stage | Build Result | Details |
|---|---|---|---|
| MindSpore-Lite_Atomgit_Gate | - | ❌ FAILURE | 1153 |
| Smoke_Android_Arm64_TF_CPU | FAILURE | 1153 |


4月16日 删除了label:ci-pipeline-running
4月16日 添加了label:ci-pipeline-failed
lilinjie11
4月16日 评论:
4月16日 评论:
/retest


4月16日 删除了label:ci-pipeline-failed
4月16日 添加了label:ci-pipeline-running
司小南(机器人)
4月16日 评论:
4月16日 评论:
| Project Name | Build_Stage | Build Result | Details |
|---|---|---|---|
| MindSpore-Lite_Atomgit_Gate | - | ✅ SUCCESS | 1155 |


4月16日 删除了label:ci-pipeline-running
4月16日 添加了label:ci-pipeline-passed
4月16日 通过审查
MindSpore-Bot
4月16日 评论:
4月16日 评论:
4月16日 删除了label:no-pass-all-review
4月16日 合入了pull request,合并节点 SHA:c9be893c4cb60a47aff673c1c22ab931163421a4
What type of PR is this?
/kind bug
What does this PR do / why do we need it:
1.权重更新在CANN更新接口变更后进行适配:权重更新出如果预期不需要输出,可以调用SetOutputs({}),不指定模型输出
修复点是:对带 is_update_graph 的图,直接不给 GE 设置任何 outputs(等价于 SetOutputs({})),从源头保证 update 模型输出数为 0,而不是依赖“删输出节点”这种隐式副作用。
原因分析:
ms lite在构图阶段,调用SetOutputs时指定了return tuple的3133个输出为模型输出,在DfGraphConvertor::RemoveIdentity中删除了模型输出(return tuple),这个Pass改变了模型的原始输出,引发问题。旧版CANN包下直接删除输出节点后模型则没有输出,但是新版CANN包中即使删除了输出节点模型的输出依然存在,所以更新CANN包后引发问题。
Which issue(s) this PR fixes:
Fixes #https://gitcode.com/mindspore/org-issues/issues/42416
Test Plan and Test result:What scenarios were tested, and what were the verification results(Function, performance, reliability, etc.):
使用0408CANN+0413lite验证通过:
Self-checklist:(请自检,在[ ]内打上x,我们将检视你的完成情况,否则会导致pr无法合入)