已合并
修复用户链接时候B库用户名查找问题 #9061
修复用户链接时候B库用户名查找问题 #9061
已合并
superlchf创建于 3月12日
superlchf成员
3月12日

【标题】修复用户链接时候B库用户名查找问题

【实现内容】:
优先获取链接用户的用户名,在找到不情况下,满足查找b库用户情况下,进行b库用户查找

【根因分析】:

【实现方案】:

【关联需求或issue】:
https://gitcode.com/opengauss/openGauss-tools-chameleon/issues/306

【开发自验报告】:

  version:
    (openGauss 7.0.0-RC3 build 0e747030) compiled at 2026-03-12 17:32:08 commit 0 last mr 9027 debug on aarch64-unknown-linux-gnu, compiled by g++ (GCC) 10.3.1, 64-bit
CONNECTOR: [IP4] pq://test_bbbbbb:***@10.175.41.120:5432/test_bbbbbb?[sslmode]=disable
  category: None
DRIVER: py_opengauss.driver.pq3.Drive
  1. 请附上自验结果(内容或者截图)
  2. 是否可以添加fastcheck测试用例,如是,请补充fastcheck用例
  3. 是否涉及资料修改,如是,在docs仓库补充资料
  4. 是否考虑升级场景(系统表修改、日志持久化以及修改执行态数据格式)
  5. 是否考虑在线扩容等扩展场景
  6. 是否考虑异常场景/并发场景/前向兼容/性能场景
  7. 是否对其他模块产生影响
  8. 是否需要回合补丁版本(5.0、6.0),涉及(结果错误|coredump|性能劣化|内存泄露|集群异常)相关修复请排查回合补丁分支

【其他说明】:

likedislike
Pull Request已成功合入, 合并人@opengauss_bot
(感谢 superlchf 的贡献)
Ssuperlchf成员
3月12日 关联了issue:[Bug]: 【工具链】【全量迁移】openGauss数据源用户名为test,数据库存在'test@%'用户,工具连接数据源报错
opengauss_botopengauss_bot成员
3月12日 添加了label:sig/StorageEnginesig/SQLEnginesig/SecurityTechnologysig/CMsig/CloudNativesig/AI
opengauss_bot
opengauss_bot成员
3月12日 评论:

Welcome To openGauss Community

Hey @superlchf , thanks for your contribution to the community.

Bot Usage Manual

I'm the Bot here serving you. You can find the instructions on how to interact with me at Here . That means you can comment below every pull request or issue to trigger Bot Commands.

Contact Guide

If you have any questions, please contact the SIG: StorageEngine, AI, CM, CloudNative, SecurityTechnology, SQLEngine ,
and any of the maintainers: @CarrotGo, @chendong76, @chenxiaobin19, @congzhou2603, @dodders, @hwworkholic, @jemappellehc, @muyulinzhong, @quemingjian, @shenzheng4, @shirley_zhengx, @superlchf, @totaj, @wlff234, @ywzq1161327784 ,
and any of the committers: @Igali, @bihua111, @cailei19, @h_ray, @levy53071, @libiao2024, @lihaixiao, @mrzack, @wangfeihuo, @wuyuechuan, @xiong_xjun, @zhangfengzhi123, @zhangxubo, @zjh_hw .

likedislike
opengauss_botopengauss_bot成员
3月12日 添加了label:codecheck-running
opengauss_botopengauss_bot成员
3月12日 添加了label:opengauss-cla/yes
opengauss_botopengauss_bot成员
3月12日 添加了label:ci-pipeline-running
opengauss_bot
opengauss_bot成员
3月12日 评论:

CLA Signature Pass

superlchf, thanks for your pull request. All authors of the commits have signed the CLA. 👍

likedislike
opengauss_botopengauss_bot成员
3月12日 添加了label:need-doc-pr
opengauss_botopengauss_bot成员
3月12日 添加了label:lite-ci-running
opengauss_bot
opengauss_bot成员
3月12日 评论:

Linking PR Notice

@superlchf , the pull request should be linked to at least one https://gitcode.com/opengauss/docs repository's PR.
If it is confirmed that no PR association is needed, you can remove the label by commenting /remove-need-doc .

likedislike
opengauss_bot
opengauss_bot成员
3月12日 评论:

开源片段扫描

Check Name Build Details Check Result Check Detail
sca #33533 ✅ 通过 >>>
likedislike
opengauss_botopengauss_bot成员
3月12日 添加了label:sca-success
opengauss_bot
opengauss_bot成员
3月12日 评论:

静态检查

Check Name Build Details Check Result Check Detail
CodeCheck #71055 ✅ 通过 >>>
likedislike
opengauss_botopengauss_bot成员
3月12日 添加了label:codecheck-success
opengauss_botopengauss_bot成员
3月12日 删除了label:codecheck-running
opengauss_botopengauss_bot成员
3月12日 删除了label:lite-ci-running
opengauss_botopengauss_bot成员
3月12日 添加了label:lite-ci-success
opengauss_bot
opengauss_bot成员
3月12日 评论:
Check Name Build Result Build Details
openGauss_Lite_Pr SUCCESS #843
likedislike
opengauss_botopengauss_bot成员
3月12日 删除了label:ci-pipeline-running
superlchf成员
3月12日 评论:

/retest

likedislike
opengauss_bot
opengauss_bot成员
3月12日 评论:
Check Name Build Result Build Details
openGauss_server_pr ABORTED #28005
likedislike
opengauss_botopengauss_bot成员
3月12日 添加了label:ci-pipeline-running
opengauss_bot
opengauss_bot成员
3月12日 评论:
Check Name Build Result Build Details
openGauss_server_pr Failed #28010
likedislike
superlchf成员
3月13日 评论:

/retest

likedislike
opengauss_botopengauss_bot成员
3月13日 删除了label:ci-pipeline-running
opengauss_botopengauss_bot成员
3月13日 添加了label:ci-pipeline-success
opengauss_bot
opengauss_bot成员
3月13日 评论:
Check Name Build Result Build Details
openGauss_server_pr Success #28016
likedislike
superlchf成员
3月13日 评论:

/remove-need-doc

likedislike
opengauss_botopengauss_bot成员
3月13日 删除了label:need-doc-pr
ywzq1161327784
ywzq1161327784成员
3月13日 评论:

/lgtm
/approve

likedislike
opengauss_botopengauss_bot成员
3月13日 添加了label:lgtm-ywzq1161327784
opengauss_bot
opengauss_bot成员
3月13日 评论:

Review Code Feedback

  • The label lgtm-ywzq1161327784 was added to this pull request. It means that ywzq1161327784 reviewed the code changes. 👋
Tips
  • If this pull request is not merged while all conditions are met, comment /check-pr to try again. 😄
likedislike
opengauss_botopengauss_bot成员
3月13日 添加了label:approved
opengauss_bot
opengauss_bot成员
3月13日 评论:

Review Code Feedback

  • The label approved was added to this pull request. It means that ywzq1161327784 reviewed the code changes. 👋
Tips
  • If this pull request is not merged while all conditions are met, comment /check-pr to try again. 😄
likedislike
wlff234
wlff234成员
3月13日 评论:

/lgtm

likedislike
opengauss_botopengauss_bot成员
3月13日 添加了label:lgtm-wlff234
opengauss_bot
opengauss_bot成员
3月13日 评论:

Review Code Feedback

  • The label lgtm-wlff234 was added to this pull request. It means that wlff234 reviewed the code changes. 👋
Tips
  • If this pull request is not merged while all conditions are met, comment /check-pr to try again. 😄
likedislike
opengauss_botopengauss_bot成员
3月13日 合入了pull request,合并节点 SHA:8339ef4ad9958febc87f0ac06c6cd0bab727ea3b
opengauss_botopengauss_bot成员
3月13日 删除了label:codecheck-success
opengauss_botopengauss_bot成员
3月13日 添加了label:codecheck-running
opengauss_bot
opengauss_bot成员
3月13日 评论:

开源片段扫描

Check Name Build Details Check Result Check Detail
sca #33597 ✅ 通过 >>>
likedislike
opengauss_bot
opengauss_bot成员
3月13日 评论:

静态检查

Check Name Build Details Check Result Check Detail
CodeCheck #71118 ✅ 通过 >>>
likedislike
opengauss_botopengauss_bot成员
3月13日 添加了label:codecheck-success
opengauss_botopengauss_bot成员
3月13日 删除了label:codecheck-running