已开启
config/tests: fix test failure in "/config/set-values" #202
AtomGit-Bot创建于 2024年11月26日
config/tests: fix test failure in "/config/set-values" #202
已开启
AtomGit-Bot创建于 2024年11月26日
refs/pull/202/head合入到openEuler-24.03-LTS-SP1
AtomGit-Bot
2024年11月26日

GKeyfile changed something about how to handle invalid escape sequences.
As we don't want to test GKeyfile (per-se), just adjust to test to not
hit the problem.

This would fail with glib2-2.79.1-1.fc40:

./tools/run-nm-test.sh -m src/core/tests/config/test-config -p /config/set-values

TAP version 13

random seed: R02Sb8afff1ec38ca5a1b7713e8c40eb4f56

Start of config tests

GLib-GIO-DEBUG: _g_io_module_get_default: Found default implementation local (GLocalVfs) for ?gio-vfs?

(src/core/tests/config/test-config.c:1107) invalid value in config-data .intern.with-whitespace.key2 = (null) (instead of " b c, d ")

./tools/run-nm-test.sh: line 307: 245847 Trace/breakpoint trap (core dumped) "NMTSTDBUSRUNSESSION[@]""{NMTST_DBUS_RUN_SESSION[@]}" "TEST" "${TEST_ARGV[@]}"
exec "src/core/tests/config/test-config" failed with exit code 133

likedislike
合并受阻
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

Hi zhong-wei-shen, welcome to the openEuler Community.
I'm the Bot here serving you. You can find the instructions on how to interact with me at Here.
If you have any questions, please contact the SIG: Networking, and any of the maintainers: @MrRlu , @zengwefeng , @sunsuwan , @robertxw , @jiangheng12 , @renmingshuai

likedislike
AtomGit-Bot
2024年11月26日 评论:

当前仓库存在以下 保护分支

Protected Branch Version Release
master 1.44.2 2
openEuler-20.03-LTS-SP4 1.26.2 14
openEuler-22.03-LTS-SP1 1.32.12 21
openEuler-22.03-LTS-SP3 1.32.12 21
openEuler-22.03-LTS-SP4 1.32.12 21
openEuler-24.03-LTS 1.44.2 3
openEuler-24.03-LTS-Next 1.44.2 3
* openEuler-24.03-LTS-SP1 1.44.2 3

评论 /sync <branch1> <branch2> ... 可将当前 PR 修改同步到其它分支(创建同步 PR):
a) 如果当前 PR 是 Open 状态,同步操作将延迟到 PR 被合并时执行
b) 如果当前 PR 已经 Merged,将立即执行同步操作

注意:

  1. /sync 命令可以指定同步到多个分支,仅最后一个 /sync 命令生效
  2. 如果创建的同步 PR 不正确,可通过向同步 PR 的源分支提交轻量级 PR 完善,或使用 /close 命令关闭
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年11月26日 添加了 &nbsp; sig/Networking 标签
openeuler-ci-botopeneuler-ci-bot成员
2024年11月26日 添加了 &nbsp; openeuler-cla/yes 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/NetworkManager/380/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年11月26日 添加了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/NetworkManager/377/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/NetworkManager/374/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年11月26日 移除了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:
Check Name Build Result Build Details
check_binary_file :white_check_mark:SUCCESS #380
check_package_yaml_file :white_check_mark:SUCCESS
check_consistency :white_check_mark:SUCCESS
check_spec_file :x:FAILED
aarch64 check_build :white_check_mark:SUCCESS #374
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
x86_64 check_build :white_check_mark:SUCCESS #377
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年11月26日 添加了 &nbsp; ci_failed 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

如下为接口变更检查结果,目标分支为openEuler-24.03-LTS-SP1,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms :white_check_mark:SUCCESS #377
delete_rpms :white_check_mark:SUCCESS
rpm_abi :white_check_mark:SUCCESS
rpm_cmd :white_check_mark:SUCCESS
rpm_config NetworkManager :x:FAILED
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_lib :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_service :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
compare_package(aarch64) add_rpms :white_check_mark:SUCCESS #374
delete_rpms :white_check_mark:SUCCESS
rpm_abi :white_check_mark:SUCCESS
rpm_cmd :white_check_mark:SUCCESS
rpm_config NetworkManager :x:FAILED
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_lib :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_service :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

@zhong-wei-shen

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 380进行误报标记,380表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 380 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 380取消
也可在评论后加上一段文字描述,但请另起一行

likedislike
AAtomGit-Bot
2024年11月26日 强制推送 了代码
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

/retest

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/NetworkManager/381/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年11月26日 移除了 &nbsp; ci_failed 标签
openeuler-ci-botopeneuler-ci-bot成员
2024年11月26日 添加了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/NetworkManager/375/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/NetworkManager/378/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年11月26日 移除了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:
Check Name Build Result Build Details
check_binary_file :white_check_mark:SUCCESS #381
check_package_yaml_file :white_check_mark:SUCCESS
check_consistency :white_check_mark:SUCCESS
check_spec_file :white_check_mark:SUCCESS
aarch64 check_build :white_check_mark:SUCCESS #375
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
x86_64 check_build :white_check_mark:SUCCESS #378
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年11月26日 添加了 &nbsp; ci_successful 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

如下为接口变更检查结果,目标分支为openEuler-24.03-LTS-SP1,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms :white_check_mark:SUCCESS #378
delete_rpms :white_check_mark:SUCCESS
rpm_abi :white_check_mark:SUCCESS
rpm_cmd :white_check_mark:SUCCESS
rpm_config NetworkManager :x:FAILED
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_lib :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_service :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
compare_package(aarch64) add_rpms :white_check_mark:SUCCESS #375
delete_rpms :white_check_mark:SUCCESS
rpm_abi :white_check_mark:SUCCESS
rpm_cmd :white_check_mark:SUCCESS
rpm_config NetworkManager :x:FAILED
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_lib :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_service :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

@zhong-wei-shen

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2024年11月26日 评论:

门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 381进行误报标记,381表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 381 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 381取消
也可在评论后加上一段文字描述,但请另起一行

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
17 天前 评论:

/check-cla

likedislike
openeuler-ci-botopeneuler-ci-bot成员
17 天前 删除了label:openeuler-cla/yes
openeuler-ci-botopeneuler-ci-bot成员
17 天前 添加了label:openeuler-cla/yes
openeuler-ci-bot
openeuler-ci-bot成员
17 天前 评论:

CLA Signature Pass

zhong-wei-shen, thanks for your pull request. All authors of the commits have signed the CLA. 👍

likedislike