已开启
更新到apptainer-1.3.3版本 #9
AtomGit-Bot创建于 2024年8月15日
更新到apptainer-1.3.3版本 #9
已开启
AtomGit-Bot创建于 2024年8月15日
从refs/pull/9/head合入到master
AtomGit-Bot
2024年8月15日

更新到apptainer-1.3.3版本

likedislike
合并受阻
AAtomGit-Bot
2024年8月15日 关联了 src-openeuler/apptainer Issue IAKBE5
openeuler-ci-botopeneuler-ci-bot成员
2024年8月15日 添加了   sig/sig-HPC 标签
openeuler-ci-botopeneuler-ci-bot成员
2024年8月15日 添加了   openeuler-cla/yes 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年8月15日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/apptainer/20/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年8月15日 添加了   ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年8月15日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/apptainer/20/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2024年8月15日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/apptainer/20/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年8月15日 移除了   ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年8月15日 评论:
Check Name Build Result Build Details
check_binary_file :white_check_mark:SUCCESS #20
check_package_yaml_file :white_check_mark:SUCCESS
check_consistency :x:FAILED
check_spec_file :x:FAILED
x86_64 check_build :x:FAILED #20
aarch64 check_build :x:FAILED #20
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年8月15日 添加了   ci_failed 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年8月15日 评论:

@brucema78

likedislike
AAtomGit-Bot
2024年8月15日 推送了代码
openeuler-ci-bot
openeuler-ci-bot成员
2024年8月15日 评论:

/retest

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年8月15日 添加了   stat/needs-squash 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年8月15日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/apptainer/21/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年8月15日 添加了   ci_processing 标签
openeuler-ci-botopeneuler-ci-bot成员
2024年8月15日 移除了   ci_failed 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年8月15日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/apptainer/21/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年8月15日 移除了   ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年8月15日 评论:
Check Name Build Result Build Details
check_binary_file :white_check_mark:SUCCESS #21
check_package_yaml_file :white_check_mark:SUCCESS
check_consistency :white_check_mark:SUCCESS
check_spec_file :x:FAILED
aarch64 check_build :white_check_mark:SUCCESS #21
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
x86_64 check_build :white_check_mark:SUCCESS #21
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2024年8月15日 添加了   ci_failed 标签
openeuler-ci-bot
openeuler-ci-bot成员
2024年8月15日 评论:

如下为接口变更检查结果,目标分支为master,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms :white_check_mark:SUCCESS #21
delete_rpms :white_check_mark:SUCCESS
rpm_cmd :white_check_mark:SUCCESS
rpm_config apptainer :x:FAILED
rpm_files apptainer :x:FAILED
rpm_provides apptainer :x:FAILED
rpm_requires apptainer
apptainer-suid
:x:FAILED
compare_package(aarch64) add_rpms :white_check_mark:SUCCESS #21
delete_rpms :white_check_mark:SUCCESS
rpm_cmd :white_check_mark:SUCCESS
rpm_config apptainer :x:FAILED
rpm_files apptainer :x:FAILED
rpm_provides apptainer :x:FAILED
rpm_requires apptainer-suid
apptainer
:x:FAILED
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2024年8月15日 评论:

@brucema78

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2024年8月15日 评论:

门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 21进行误报标记,21表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 21 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 21取消
也可在评论后加上一段文字描述,但请另起一行

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
8月22日 评论:

/check-cla

likedislike
openeuler-ci-botopeneuler-ci-bot成员
8月22日 删除了label:openeuler-cla/yes
openeuler-ci-botopeneuler-ci-bot成员
8月22日 添加了label:openeuler-cla/yes
openeuler-ci-bot
openeuler-ci-bot成员
8月22日 评论:

CLA Signature Pass

brucema78, thanks for your pull request. All authors of the commits have signed the CLA. 👍

likedislike