已开启
backport patches for fixing Wincompatible-pointer-types when building with newer compilers #194
jchzhou创建于 2025年8月27日
backport patches for fixing Wincompatible-pointer-types when building with newer compilers #194
已开启
jchzhou创建于 2025年8月27日
从refs/pull/194/head合入到openEuler-25.09
jchzhou
2025年8月27日
  1. backport patches for fixing Wincompatible-pointer-types when building with newer compilers

Fix introduced warnings
https://github.com/linux-audit/audit-userspace/commit/b8479df52ce3f286ba010268527c9c280abaf478

likedislike
合并受阻
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月27日 评论:

Hi jchzhou, welcome to the openEuler Community.
I'm the Bot here serving you. You can find the instructions on how to interact with me at Here.
If you have any questions, please contact the SIG: sig-security-facility, and any of the maintainers: @zhujianwei001 , @mailofzxf , @robertosassu , @gwei3 , @wucaijun2001 , any of the committers: @huangzq6

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年8月27日 指派了 zhujianwei001 参与评审
openeuler-ci-botopeneuler-ci-bot成员
2025年8月27日 指派了 gwei3 参与评审
openeuler-ci-botopeneuler-ci-bot成员
2025年8月27日 指派了 wucaijun 参与评审
openeuler-ci-botopeneuler-ci-bot成员
2025年8月27日 指派了 mailofzxf 参与评审
openeuler-ci-botopeneuler-ci-bot成员
2025年8月27日 指派了 robertosassu 参与评审
AtomGit-Bot
2025年8月27日 评论:

当前仓库存在以下 保护分支 :

Protected Branch Version Release
master 3.1.2 11
openEuler-20.03-LTS-SP4 3.0 16
openEuler-22.03-LTS-SP3 3.0.1 20
openEuler-22.03-LTS-SP4 3.0.1 20
openEuler-24.03-LTS 3.1.2 11
openEuler-24.03-LTS-Next 3.1.2 12
openEuler-24.03-LTS-SP1 3.1.2 11
openEuler-24.03-LTS-SP2 3.1.2 11
openEuler-25.03 3.1.2 10
* openEuler-25.09 3.1.2 12

评论 /sync <branch1> <branch2> ... 可将当前 PR 修改同步到其它分支(创建同步 PR):
a) 如果当前 PR 是 Open 状态,同步操作将延迟到 PR 被合并时执行
b) 如果当前 PR 已经 Merged,将立即执行同步操作

注意:

  1. /sync 命令可以指定同步到多个分支,仅最后一个 /sync 命令生效
  2. 如果创建的同步 PR 不正确,可通过向同步 PR 的源分支提交轻量级 PR 完善,或使用 /close 命令关闭
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年8月27日 添加了 &nbsp; openeuler-cla/yes 标签
openeuler-ci-botopeneuler-ci-bot成员
2025年8月27日 添加了 &nbsp; sig/sig-security-fac 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月27日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/audit/305/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年8月27日 添加了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月27日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/audit/305/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月27日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/audit/305/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年8月27日 移除了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月27日 评论:
Check Name Build Result Build Details
check_binary_file :white_check_mark:SUCCESS #305
check_package_yaml_file :white_check_mark:SUCCESS
check_repo_in_maintain :white_check_mark:SUCCESS
check_consistency :white_check_mark:SUCCESS
check_spec_file :white_check_mark:SUCCESS
aarch64 check_build :white_check_mark:SUCCESS #305
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
x86_64 check_build :white_check_mark:SUCCESS #305
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年8月27日 添加了 &nbsp; ci_successful 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月27日 评论:

@jchzhou

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月27日 评论:

如下为接口变更检查结果,目标分支为openEuler-25.09,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms :white_check_mark:SUCCESS #305
delete_rpms :white_check_mark:SUCCESS
rpm_abi :white_check_mark:SUCCESS
rpm_cmd :white_check_mark:SUCCESS
rpm_config :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_lib :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_service :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
compare_package(aarch64) add_rpms :white_check_mark:SUCCESS #305
delete_rpms :white_check_mark:SUCCESS
rpm_abi :white_check_mark:SUCCESS
rpm_cmd :white_check_mark:SUCCESS
rpm_config :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_lib :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_service :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月27日 评论:

门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 305进行误报标记,305表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 305 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 305取消
也可在评论后加上一段文字描述,但请另起一行

likedislike
xuraoqing2025年10月9日进行代码检视1
audit.spec
@@ -2,7 +2,7 @@ Summary: User space tools for kernel auditing
22Name: audit
33Epoch: 1
44Version: 3.1.2
5-Release: 12
5+Release: 13
66License: GPLv2+ and LGPLv2+
77URL: https://people.redhat.com/sgrubb/audit/
88Source0: https://people.redhat.com/sgrubb/audit/%{name}-%{version}.tar.gz
@@ -65,6 +65,8 @@ Patch51: backport-make-slist_last-a-static-function.patch
6565Patch52: backport-code-cleanup-constification-deadcode-removal-lower-s.patch
6666# https://github.com/linux-audit/audit-userspace/commit/8c7eaa7ead6c70486623674c19d649f3831578ad
6767Patch53: backport-Implicit-builtin-functions.patch
68+# https://github.com/linux-audit/audit-userspace/commit/b8479df52ce3f286ba010268527c9c280abaf478
xuraoqing2025年10月9日评论:

建议在补丁头中包含社区提交链接,例如:
backport-Fix-deprecated-python-function.patch

likedislike
jchzhou2025年10月9日进行代码检视1
audit.spec
@@ -2,7 +2,7 @@ Summary: User space tools for kernel auditing
22Name: audit
33Epoch: 1
44Version: 3.1.2
5-Release: 12
5+Release: 13
66License: GPLv2+ and LGPLv2+
77URL: https://people.redhat.com/sgrubb/audit/
88Source0: https://people.redhat.com/sgrubb/audit/%{name}-%{version}.tar.gz
@@ -65,6 +65,8 @@ Patch51: backport-make-slist_last-a-static-function.patch
6565Patch52: backport-code-cleanup-constification-deadcode-removal-lower-s.patch
6666# https://github.com/linux-audit/audit-userspace/commit/8c7eaa7ead6c70486623674c19d649f3831578ad
6767Patch53: backport-Implicit-builtin-functions.patch
68+# https://github.com/linux-audit/audit-userspace/commit/b8479df52ce3f286ba010268527c9c280abaf478
jchzhou2025年10月9日评论:

建议在补丁头中包含社区提交链接,例如:
backport-Fix-deprecated-python-function.patch

收到,这就对应修改

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年10月10日 评论:

/retest

likedislike
Jjchzhou
2025年10月10日 强制推送 了代码
openeuler-ci-bot
openeuler-ci-bot成员
2025年10月10日 评论:

This pull request source branch has changed, so removes the following label(s): ci_successful.

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年10月10日 移除了 &nbsp; ci_successful 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年10月10日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/audit/310/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年10月10日 添加了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年10月10日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/audit/310/console

likedislike
jchzhou2025年10月10日进行代码检视1
audit.spec
@@ -2,7 +2,7 @@ Summary: User space tools for kernel auditing
22Name: audit
33Epoch: 1
44Version: 3.1.2
5-Release: 12
5+Release: 13
66License: GPLv2+ and LGPLv2+
77URL: https://people.redhat.com/sgrubb/audit/
88Source0: https://people.redhat.com/sgrubb/audit/%{name}-%{version}.tar.gz
@@ -65,6 +65,8 @@ Patch51: backport-make-slist_last-a-static-function.patch
6565Patch52: backport-code-cleanup-constification-deadcode-removal-lower-s.patch
6666# https://github.com/linux-audit/audit-userspace/commit/8c7eaa7ead6c70486623674c19d649f3831578ad
6767Patch53: backport-Implicit-builtin-functions.patch
68+# https://github.com/linux-audit/audit-userspace/commit/b8479df52ce3f286ba010268527c9c280abaf478
jchzhou2025年10月10日评论:

建议在补丁头中包含社区提交链接,例如:
backport-Fix-deprecated-python-function.patch

感谢指点,已经修改完成

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年10月10日 评论:
Check Name Build Result Build Details
check_consistency :white_check_mark:SUCCESS #310
check_binary_file :white_check_mark:SUCCESS
check_repo_in_maintain :white_check_mark:SUCCESS
check_spec_file :white_check_mark:SUCCESS
check_package_yaml_file :white_check_mark:SUCCESS
aarch64 check_build :white_check_mark:SUCCESS #310
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
x86_64 check_build :white_check_mark:SUCCESS #310
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年10月10日 移除了 &nbsp; ci_processing 标签
openeuler-ci-botopeneuler-ci-bot成员
2025年10月10日 添加了 &nbsp; ci_successful 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年10月10日 评论:

如下为接口变更检查结果,目标分支为openEuler-25.09,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms :white_check_mark:SUCCESS #310
delete_rpms :white_check_mark:SUCCESS
rpm_abi :white_check_mark:SUCCESS
rpm_cmd :white_check_mark:SUCCESS
rpm_config :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_lib :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_service :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
compare_package(aarch64) add_rpms :white_check_mark:SUCCESS #310
delete_rpms :white_check_mark:SUCCESS
rpm_abi :white_check_mark:SUCCESS
rpm_cmd :white_check_mark:SUCCESS
rpm_config :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_lib :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_service :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年10月10日 评论:

@jchzhou

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年10月10日 评论:

门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 310进行误报标记,310表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 310 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 310取消
也可在评论后加上一段文字描述,但请另起一行

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
8月22日 评论:

/check-cla

likedislike
openeuler-ci-botopeneuler-ci-bot成员
8月22日 删除了label:openeuler-cla/yes
openeuler-ci-botopeneuler-ci-bot成员
8月22日 添加了label:openeuler-cla/yes
openeuler-ci-bot
openeuler-ci-bot成员
8月22日 评论:

CLA Signature Pass

jchzhou, thanks for your pull request. All authors of the commits have signed the CLA. 👍

likedislike