已开启
[sync] PR-38: remove unused patch1 #42
AtomGit-Bot创建于 2025年4月7日
[sync] PR-38: remove unused patch1 #42
已开启
AtomGit-Bot创建于 2025年4月7日
AtomGit-Bot
2025年4月7日

1. Origin pull request:

https://gitee.com/src-openeuler/bazel/pulls/38

Sha Datetime Message
79a47b43 2025-02-19 11:38:20 +0800 CST remove unused patch1
likedislike
合并受阻
openeuler-ci-bot
openeuler-ci-bot成员
2025年4月7日 评论:

Hi openeuler-sync-bot, welcome to the openEuler Community.
I'm the Bot here serving you. You can find the instructions on how to interact with me at Here.
If you have any questions, please contact the SIG: ai, and any of the maintainers: @jimmy_hero , @weixizhu94 , @sinever , @zhushoukui , @anhongkui

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年4月7日 添加了   sig/ai 标签
openeuler-ci-botopeneuler-ci-bot成员
2025年4月7日 添加了   openeuler-cla/yes 标签
AtomGit-Bot
2025年4月7日 评论:

/check-cla

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年4月7日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/bazel/82/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年4月7日 添加了   ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年4月7日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/bazel/81/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年4月7日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/bazel/80/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年4月7日 移除了   ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年4月7日 评论:
Check Name Build Result Build Details
check_binary_file :white_check_mark:SUCCESS #82
check_package_yaml_file :white_check_mark:SUCCESS
check_consistency :white_check_mark:SUCCESS
check_spec_file :bug:WARNING
aarch64 check_build :white_check_mark:SUCCESS #80
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
x86_64 check_build :white_check_mark:SUCCESS #81
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年4月7日 添加了   ci_successful 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年4月7日 评论:

@openeuler-sync-bot

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年4月7日 评论:

如下为接口变更检查结果,目标分支为master,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms :white_check_mark:SUCCESS #81
delete_rpms :white_check_mark:SUCCESS
rpm_cmd :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
compare_package(aarch64) add_rpms :white_check_mark:SUCCESS #80
delete_rpms :white_check_mark:SUCCESS
rpm_cmd :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年4月7日 评论:

门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 82进行误报标记,82表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 82 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 82取消
也可在评论后加上一段文字描述,但请另起一行

likedislike
laokz
laokz
2月25日 评论:

ping...

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
29 天前 评论:

/check-cla

likedislike
openeuler-ci-botopeneuler-ci-bot成员
29 天前 删除了label:openeuler-cla/yes
openeuler-ci-botopeneuler-ci-bot成员
29 天前 添加了label:openeuler-cla/yes
openeuler-ci-bot
openeuler-ci-bot成员
29 天前 评论:

CLA Signature Pass

openeuler-sync-bot, thanks for your pull request. All authors of the commits have signed the CLA. 👍

likedislike
eaglegai
18 天前 评论:

/retest

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
18 天前 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/bazel/8/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
18 天前 删除了label:ci_successful
openeuler-ci-botopeneuler-ci-bot成员
18 天前 添加了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
18 天前 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/bazel/8/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
18 天前 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/bazel/8/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
18 天前 删除了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
18 天前 评论:
Check Name Build Result 详情 Build Details
check_binary_file SUCCESS #8
check_lfsconfig :ballot_box_with_check:EXCLUDE
check_package_yaml_file SUCCESS
check_repo_in_maintain SUCCESS
check_consistency SUCCESS
check_spec_file SUCCESS
x86_64 check_build SUCCESS #8
check_install SUCCESS
check_license SUCCESS
aarch64 check_build SUCCESS #8
check_install SUCCESS
check_license SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
18 天前 添加了label:ci_successful
openeuler-ci-bot
openeuler-ci-bot成员
18 天前 评论:

如下为接口变更检查结果,目标分支为master,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms SUCCESS #8
delete_rpms SUCCESS
rpm_cmd SUCCESS
rpm_files SUCCESS
rpm_provides SUCCESS
rpm_requires bazel FAILED
rpm_symbol SUCCESS
compare_package(aarch64) add_rpms SUCCESS #8
delete_rpms SUCCESS
rpm_cmd SUCCESS
rpm_files SUCCESS
rpm_provides SUCCESS
rpm_requires bazel FAILED
rpm_symbol SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
18 天前 评论:
likedislike