已开启
backport net/http: raise an error when a http2 header frame is malformed #558
fortunate-lee创建于 2025年11月11日
backport net/http: raise an error when a http2 header frame is malformed #558
已开启
fortunate-lee创建于 2025年11月11日
从refs/pull/558/head合入到openEuler-24.03-LTS-SP2
fortunate-lee
fortunate-lee
2025年11月11日

net/http: raise an error when a http2 header frame is malformed.

likedislike
合并受阻
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月11日 评论:

Hi fortunate-lee, welcome to the openEuler Community.
I'm the Bot here serving you. You can find the instructions on how to interact with me at Here.
If you have any questions, please contact the SIG: sig-golang, and any of the maintainers: @jing-rui , @genedna , @hcnbxx , any of the committers: @fuowang , @wangding16

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年11月11日 指派了 jing-rui 参与评审
openeuler-ci-botopeneuler-ci-bot成员
2025年11月11日 指派了 hc 参与评审
openeuler-ci-botopeneuler-ci-bot成员
2025年11月11日 指派了 genedna 参与评审
AtomGit-Bot
2025年11月11日 评论:

当前仓库存在以下 保护分支 :

Protected Branch Version Release
master 1.24.2 39
openEuler-20.03-LTS-SP4 1.15.7 54
openEuler-22.03-LTS-SP3 1.17.3 42
openEuler-22.03-LTS-SP4 1.17.3 42
openEuler-24.03-LTS 1.21.4 37
openEuler-24.03-LTS-Next 1.21.4 37
openEuler-24.03-LTS-SP1 1.21.4 37
* openEuler-24.03-LTS-SP2 1.21.4 37
openEuler-24.03-LTS-SP3 1.21.4 39
openEuler-25.03 1.21.4 32
openEuler-25.09 1.24.2 36

评论 /sync <branch1> <branch2> ... 可将当前 PR 修改同步到其它分支(创建同步 PR):
a) 如果当前 PR 是 Open 状态,同步操作将延迟到 PR 被合并时执行
b) 如果当前 PR 已经 Merged,将立即执行同步操作

注意:

  1. /sync 命令可以指定同步到多个分支,仅最后一个 /sync 命令生效
  2. 如果创建的同步 PR 不正确,可通过向同步 PR 的源分支提交轻量级 PR 完善,或使用 /close 命令关闭
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年11月11日 添加了 &nbsp; sig/sig-golang 标签
openeuler-ci-botopeneuler-ci-bot成员
2025年11月11日 添加了 &nbsp; openeuler-cla/yes 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月11日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/golang/1043/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年11月11日 添加了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月11日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/golang/1020/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月11日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/golang/1019/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月11日 评论:
Check Name Build Result Build Details
check_binary_file :white_check_mark:SUCCESS #1043
check_package_yaml_file :white_check_mark:SUCCESS
check_repo_in_maintain :white_check_mark:SUCCESS
check_consistency :white_check_mark:SUCCESS
check_spec_file :x:FAILED
x86_64 check_build :white_check_mark:SUCCESS #1020
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
aarch64 check_build :white_check_mark:SUCCESS #1019
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年11月11日 移除了 &nbsp; ci_processing 标签
openeuler-ci-botopeneuler-ci-bot成员
2025年11月11日 添加了 &nbsp; ci_failed 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月11日 评论:

如下为接口变更检查结果,目标分支为openEuler-24.03-LTS-SP2,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms :white_check_mark:SUCCESS #1020
delete_rpms :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
compare_package(aarch64) add_rpms :white_check_mark:SUCCESS #1019
delete_rpms :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月11日 评论:

@fortunate-lee

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月11日 评论:

门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 1043进行误报标记,1043表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 1043 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 1043取消
也可在评论后加上一段文字描述,但请另起一行

likedislike
fuowang成员
2025年11月12日 评论:

ci失败,请修改spec文件

likedislike
fortunate-leefortunate-lee
2025年11月12日 强制推送 了代码
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

/retest

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/golang/1055/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年11月12日 移除了 &nbsp; ci_failed 标签
openeuler-ci-botopeneuler-ci-bot成员
2025年11月12日 添加了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/golang/1032/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/golang/1031/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年11月12日 移除了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:
Check Name Build Result Build Details
check_binary_file :white_check_mark:SUCCESS #1055
check_package_yaml_file :white_check_mark:SUCCESS
check_repo_in_maintain :white_check_mark:SUCCESS
check_consistency :white_check_mark:SUCCESS
check_spec_file :white_check_mark:SUCCESS
x86_64 check_build :white_check_mark:SUCCESS #1032
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
aarch64 check_build :white_check_mark:SUCCESS #1031
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

如下为接口变更检查结果,目标分支为openEuler-24.03-LTS-SP2,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms :white_check_mark:SUCCESS #1032
delete_rpms :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
compare_package(aarch64) add_rpms :white_check_mark:SUCCESS #1031
delete_rpms :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

@fortunate-lee

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 1055进行误报标记,1055表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 1055 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 1055取消
也可在评论后加上一段文字描述,但请另起一行

likedislike
fuowang成员2025年11月12日进行代码检视1
backport-0037-net-http-raise-an-error-when-a-http2-header-frame-is.patch
@@ -0,0 +1,2 @@
1+From 987589dc7dbfe5967f3cddf81dd298e91fd3c85b Mon Sep 17 00:00:00 2001
2+From: Jonathan KUMA <jonathan.kuma@gmail.com>
3+Date: Thu, 6 Nov 2025 14:51:41 +0100
fuowang2025年11月12日评论:

上游补丁,请保留原始patch头部信息

likedislike
fortunate-leefortunate-lee
2025年11月12日 强制推送 了代码
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

/retest

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

This pull request source branch has changed, so removes the following label(s): ci_successful.

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/golang/1056/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年11月12日 添加了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/golang/1032/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/golang/1033/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年11月12日 移除了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:
Check Name Build Result Build Details
check_binary_file :white_check_mark:SUCCESS #1056
check_package_yaml_file :white_check_mark:SUCCESS
check_repo_in_maintain :white_check_mark:SUCCESS
check_consistency :white_check_mark:SUCCESS
check_spec_file :white_check_mark:SUCCESS
x86_64 check_build :white_check_mark:SUCCESS #1033
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
aarch64 check_build :white_check_mark:SUCCESS #1032
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年11月12日 添加了 &nbsp; ci_successful 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

如下为接口变更检查结果,目标分支为openEuler-24.03-LTS-SP2,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms :white_check_mark:SUCCESS #1033
delete_rpms :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
compare_package(aarch64) add_rpms :white_check_mark:SUCCESS #1032
delete_rpms :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_header :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
rpm_symbol :white_check_mark:SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

@fortunate-lee

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年11月12日 评论:

门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 1056进行误报标记,1056表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 1056 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 1056取消
也可在评论后加上一段文字描述,但请另起一行

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
8月22日 评论:

/check-cla

likedislike
openeuler-ci-botopeneuler-ci-bot成员
8月22日 删除了label:openeuler-cla/yes
openeuler-ci-botopeneuler-ci-bot成员
8月22日 添加了label:openeuler-cla/yes
openeuler-ci-bot
openeuler-ci-bot成员
8月22日 评论:

CLA Signature Pass

fortunate-lee, thanks for your pull request. All authors of the commits have signed the CLA. 👍

likedislike