合并受阻
Hi konglidong, welcome to the openEuler Community.
I'm the Bot here serving you. You can find the instructions on how to interact with me at Here.
If you have any questions, please contact the SIG: Desktop, and any of the maintainers: @wangxp006 , @small_leek , @yanan-rock , @t_feng , @dou33 , @dwl301 , @weidongkl , @open-bot , @Lostwayzxc


当前仓库存在以下 保护分支 :
| Protected Branch | Version | Release |
|---|---|---|
| * master | 2.6.1 | 1 |
| openEuler-20.03-LTS-SP1 | 2.2.0 | 3 |
| openEuler-20.03-LTS-SP3 | 2.2.0 | 3 |
| openEuler-20.03-LTS-SP4 | 2.2.0 | 3 |
| openEuler-22.03-LTS | 2.4.0 | 1 |
| openEuler-22.03-LTS-Next | 2.5.1 | 1 |
| openEuler-22.03-LTS-SP1 | 2.5.1 | 1 |
| openEuler-22.03-LTS-SP2 | 2.5.1 | 1 |
| openEuler-22.03-LTS-SP3 | 2.5.1 | 1 |
| openEuler-23.09 | 2.6.1 | 1 |
评论 /sync <branch1> <branch2> ... 可将当前 PR 修改同步到其它分支(创建同步 PR):
a) 如果当前 PR 是 Open 状态,同步操作将延迟到 PR 被合并时执行
b) 如果当前 PR 已经 Merged,将立即执行同步操作
注意:
- /sync 命令可以指定同步到多个分支,仅最后一个 /sync 命令生效
- 如果创建的同步 PR 不正确,可通过向同步 PR 的源分支提交轻量级 PR 完善,或使用 /close 命令关闭


门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/kbd, 当前构建号为 56
构建及构建后检查: multiarch/src-openeuler/x86-64/kbd, multiarch/src-openeuler/aarch64/kbd
门禁结果回显: multiarch/src-openeuler/comment/kbd
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册


| Check Name | Build Result | Build Details | |
|---|---|---|---|
| check_binary_file | :white_check_mark:SUCCESS | #56 | |
| check_package_license | :white_check_mark:SUCCESS | ||
| check_package_yaml_file | :white_check_mark:SUCCESS | ||
| check_consistency | :bug:WARNING | ||
| check_spec_file | :bug:WARNING | ||
| aarch64 | check_build | :x:FAILED | #56 |
| x86_64 | check_build | :x:FAILED | #56 |


@konglidong


门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 56进行误报标记,56表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 56 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 56取消
也可在评论后加上一段文字描述,但请另起一行


/retest


门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/kbd, 当前构建号为 57
构建及构建后检查: multiarch/src-openeuler/x86-64/kbd, multiarch/src-openeuler/aarch64/kbd
门禁结果回显: multiarch/src-openeuler/comment/kbd
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册


@konglidong


门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 57进行误报标记,57表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 57 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 57取消
也可在评论后加上一段文字描述,但请另起一行


@孔立栋 门禁失败了,应该是补丁制作的有点问题,谢谢


/retest


门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/kbd, 当前构建号为 59
构建及构建后检查: multiarch/src-openeuler/x86-64/kbd, multiarch/src-openeuler/aarch64/kbd
门禁结果回显: multiarch/src-openeuler/comment/kbd
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册


@konglidong


门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 59进行误报标记,59表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 59 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 59取消
也可在评论后加上一段文字描述,但请另起一行


/retest


门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/kbd/69/console


x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/kbd/69/console


aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/kbd/69/console


@konglidong


门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 69进行误报标记,69表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 69 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 69取消
也可在评论后加上一段文字描述,但请另起一行


/check-cla


CLA Signature Guide
@konglidong , thanks for your pull request.
The following commit(s) are not associated with a signed Contributor License Agreement (CLA).
| Commit | Reason |
|---|---|
| c78bf707 update version to 2.6.4 | the email used in the commit is not linked to a signed CLA! please verify that it matches the email you used when signing the CLA. |
To sign CLA, click here.
To check if your email is configured correctly, refer to the FAQs.
Once you've signed the CLA or updating your email, please comment /check-cla to revalidate CLA status.


#I8PSCC