已关闭
kernel: Split extra modules into kernel-extra-modules subpackage #2172
lw5202031创建于 2025年8月4日关闭于 4 天前
kernel: Split extra modules into kernel-extra-modules subpackage #2172
已关闭
lw5202031创建于 2025年8月4日关闭于 4 天前
refs/pull/2172/head合入到master
lw5202031成员
2025年8月4日

This commit introduces the following changes:

Generates dedicated filelists:
modules-core.list for kernel-core
modules-extra.list for kernel-extra-modules

Adds kernel-extra-modules subpackage definitions in the .spec file.
Motivation:
Allow optional installation of niche/third-party drivers
Align with upstream modular packaging practices

TC会议纪要:https://etherpad.openeuler.org/p/TC-meetings

likedislike
当前Pull Request已关闭, 关闭人@lw5202031
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:

Hi lw520203, welcome to the openEuler Community.
I'm the Bot here serving you. You can find the instructions on how to interact with me at Here.
If you have any questions, please contact the SIG: Kernel, and any of the maintainers.

likedislike
Llw5202031成员
2025年8月4日 指派了 zhangchangzhong 参与评审
Llw5202031成员
2025年8月4日 指派了 sanglipeng 参与评审
openeuler-ci-botopeneuler-ci-bot成员
2025年8月4日 添加了   sig/Kernel 标签
openeuler-ci-botopeneuler-ci-bot成员
2025年8月4日 添加了   openeuler-cla/yes 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/kernel/4107/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年8月4日 添加了   ci_processing 标签
AtomGit-Bot
2025年8月4日 评论:

当前仓库存在以下 保护分支

Protected Branch Version Release
* master 6.12.33 0.0.0.1
openEuler-20.03-LTS-SP4 4.19.90 2507.5.0.0338
openEuler-22.03-LTS-SP3 5.10.0 275.0.0.177%{?extra_release}
openEuler-22.03-LTS-SP4 5.10.0 275.0.0.178%{?extra_release}
openEuler-24.03-LTS 6.6.0 103.0.0.95
openEuler-24.03-LTS-Next 6.6.0 84.0.0.89
openEuler-24.03-LTS-SP1 6.6.0 103.0.0.106
openEuler-24.03-LTS-SP2 6.6.0 103.0.0.109
openEuler-25.03 6.6.0 72.6.0.56
openEuler-25.09 6.6.0 102.0.0.2

评论 /sync <branch1> <branch2> ... 可将当前 PR 修改同步到其它分支(创建同步 PR):
a) 如果当前 PR 是 Open 状态,同步操作将延迟到 PR 被合并时执行
b) 如果当前 PR 已经 Merged,将立即执行同步操作

注意:

  1. /sync 命令可以指定同步到多个分支,仅最后一个 /sync 命令生效
  2. 如果创建的同步 PR 不正确,可通过向同步 PR 的源分支提交轻量级 PR 完善,或使用 /close 命令关闭
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/kernel/4132/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/kernel/4188/console

likedislike
Llw5202031成员
2025年8月4日 强制推送 了代码
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:

/retest

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/kernel/4108/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/kernel/4189/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/kernel/4133/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年8月4日 移除了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:
Check Name Build Result Build Details
check_binary_file :white_check_mark:SUCCESS #4107
check_package_yaml_file :white_check_mark:SUCCESS
check_patch_format :white_check_mark:SUCCESS
check_repo_in_maintain :white_check_mark:SUCCESS
check_consistency :white_check_mark:SUCCESS
check_spec_file :white_check_mark:SUCCESS
aarch64 check_build :x:FAILED #4188
x86_64 check_build :x:FAILED #4132
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:

@lw520203

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年8月4日 添加了 &nbsp; ci_failed 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:

门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 4107进行误报标记,4107表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 4107 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 4107取消
也可在评论后加上一段文字描述,但请另起一行

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:
Check Name Build Result Build Details
check_binary_file :white_check_mark:SUCCESS #4108
check_package_yaml_file :white_check_mark:SUCCESS
check_patch_format :white_check_mark:SUCCESS
check_repo_in_maintain :white_check_mark:SUCCESS
check_consistency :white_check_mark:SUCCESS
check_spec_file :white_check_mark:SUCCESS
aarch64 check_build :white_check_mark:SUCCESS #4189
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
x86_64 check_build :x:FAILED #4133
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:

@lw520203

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年8月4日 评论:

门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 4108进行误报标记,4108表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 4108 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 4108取消
也可在评论后加上一段文字描述,但请另起一行

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
16 天前 评论:

/check-cla

likedislike
openeuler-ci-botopeneuler-ci-bot成员
16 天前 删除了label:openeuler-cla/yes
openeuler-ci-botopeneuler-ci-bot成员
16 天前 添加了label:openeuler-cla/yes
openeuler-ci-bot
openeuler-ci-bot成员
16 天前 评论:

CLA Signature Pass

lw520203, thanks for your pull request. All authors of the commits have signed the CLA. 👍

likedislike
Llw5202031成员
4 天前 关闭了 pull request