已开启
fix CVE-2025-7425 #383
fix CVE-2025-7425 #383
已开启
gao-yu-kun创建于 3月10日
gao-yu-kun
3月10日

#92 CVE-2025-7425

likedislike
合并受阻
Ggao-yu-kun
3月10日 创建了 pull request,commit 8ec338ef
Ggao-yu-kun
3月10日 关联了issue:CVE-2025-7425
openeuler-ci-botopeneuler-ci-bot成员
3月10日 将licihua,dillon_chen,overweight,zhujianwei001,shenyangyang设为审查人
openeuler-ci-botopeneuler-ci-bot成员
3月10日 添加了label:sig/Base-service
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:

Welcome To openEuler Community

Hey @gao-yu-kun , thanks for your contribution to the community.

Bot Usage Manual

I'm the Bot here serving you. You can find the instructions on how to interact with me at Here . That means you can comment below every pull request or issue to trigger Bot Commands.

Contact Guide

If you have any questions, please contact the SIG: Base-service ,
and any of the maintainers: @dillon_chen, @licihua, @overweight, @shenyangyang, @zhujianwei001 ,
and any of the committers: @hubin95, @luckky7, @znzju .

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:

当前仓库存在以下 保护分支 :

Protected Branch Version Release
master 2.15.2 1
openEuler-24.03-LTS-SP3 2.11.9 12
openEuler-24.03-LTS-SP2 2.11.9 12
openEuler-24.03-LTS-SP1 2.11.9 12
openEuler-24.03-LTS 2.11.5 15
openEuler-22.03-LTS-SP4 2.9.14 26
openEuler-20.03-LTS-SP4 2.9.10 55
openEuler-24.03-LTS-Next 2.11.9 12
openEuler-25.09 2.12.10 6
openEuler-25.03 2.12.10 3
revert-merge-224-master 2.11.5 2
openEuler1.0-base 2.9.8 8
openEuler1.0 2.9.8 8

评论 /sync <branch1> <branch2> ... 可将当前 PR 修改同步到其它分支(创建同步 PR):
a) 如果当前 PR 是 Open 状态,同步操作将延迟到 PR 被合并时执行
b) 如果当前 PR 已经 Merged,将立即执行同步操作

注意:

  1. /sync 命令可以指定同步到多个分支,仅最后一个 /sync 命令生效
  2. 如果创建的同步 PR 不正确,可通过向同步 PR 的源分支提交轻量级 PR 完善,或使用 /close 命令关闭
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/libxml2/64/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月10日 添加了label:openeuler-cla/yes
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:

CLA Signature Pass

gao-yu-kun, thanks for your pull request. All authors of the commits have signed the CLA. 👍

likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月10日 添加了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/libxml2/64/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/libxml2/64/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月10日 删除了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:
Check Name Build Result Build Details
check_binary_file ✅SUCCESS #64
check_package_yaml_file ✅SUCCESS
check_repo_in_maintain ✅SUCCESS
check_consistency ✅SUCCESS
check_spec_file ✅SUCCESS
x86_64 check_build ✅SUCCESS #64
check_install ✅SUCCESS
check_license ✅SUCCESS
aarch64 check_build ✅SUCCESS #64
check_install ✅SUCCESS
check_license ✅SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月10日 添加了label:ci_successful
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:

如下为接口变更检查结果,目标分支为openEuler-24.03-LTS,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms ✅SUCCESS #64
delete_rpms ✅SUCCESS
rpm_abi ✅SUCCESS
rpm_cmd ✅SUCCESS
rpm_files ✅SUCCESS
rpm_header ✅SUCCESS
rpm_lib ✅SUCCESS
rpm_provides ✅SUCCESS
rpm_requires ✅SUCCESS
rpm_symbol ✅SUCCESS
compare_package(aarch64) add_rpms ✅SUCCESS #64
delete_rpms ✅SUCCESS
rpm_abi ✅SUCCESS
rpm_cmd ✅SUCCESS
rpm_files ✅SUCCESS
rpm_header ✅SUCCESS
rpm_lib ✅SUCCESS
rpm_provides ✅SUCCESS
rpm_requires ✅SUCCESS
rpm_symbol ✅SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:
likedislike
xiaoniuzi
xiaoniuzi
3月10日 评论:

/retest

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/libxml2/65/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月10日 删除了label:ci_successful
openeuler-ci-botopeneuler-ci-bot成员
3月10日 添加了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/libxml2/65/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/libxml2/65/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月10日 删除了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:
Check Name Build Result Build Details
check_binary_file ✅SUCCESS #65
check_package_yaml_file ✅SUCCESS
check_repo_in_maintain ✅SUCCESS
check_consistency ✅SUCCESS
check_spec_file ✅SUCCESS
x86_64 check_build ✅SUCCESS #65
check_install ✅SUCCESS
check_license ✅SUCCESS
aarch64 check_build ✅SUCCESS #65
check_install ✅SUCCESS
check_license ✅SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月10日 添加了label:ci_successful
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:

如下为接口变更检查结果,目标分支为openEuler-24.03-LTS,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms ✅SUCCESS #65
delete_rpms ✅SUCCESS
rpm_abi ✅SUCCESS
rpm_cmd ✅SUCCESS
rpm_files ✅SUCCESS
rpm_header ✅SUCCESS
rpm_lib ✅SUCCESS
rpm_provides ✅SUCCESS
rpm_requires ✅SUCCESS
rpm_symbol ✅SUCCESS
compare_package(aarch64) add_rpms ✅SUCCESS #65
delete_rpms ✅SUCCESS
rpm_abi ✅SUCCESS
rpm_cmd ✅SUCCESS
rpm_files ✅SUCCESS
rpm_header ✅SUCCESS
rpm_lib ✅SUCCESS
rpm_provides ✅SUCCESS
rpm_requires ✅SUCCESS
rpm_symbol ✅SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月10日 评论:
likedislike
zhuo
3月18日 评论:

这个是其他发行版的补丁,上游社区明确说明这会造成api滥用,一切已上游社区为主,请不要合入

likedislike
zhuo
3月18日 评论:

/close

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月18日 评论:

Notice

@zhuofeng6 , you can't close a pull request unless you are the author of it or a collaborator.

likedislike
gao-yu-kun
3月19日 评论:

这个是其他发行版的补丁,上游社区明确说明这会造成api滥用,一切已上游社区为主,请不要合入

@zhuofeng6

漏洞根因是 libxslt 对 libxml2 内部字段的滥用,并非是补丁api滥用

likedislike
陈棋德
陈棋德成员
3月23日 评论:

libxml2原maintainer不再维护由新来的人接手
上游社区这个issue还未关闭,同意暂不合入

看了代码,从代码逻辑来看并没有改变,感觉从代码来看是可以合入的

likedislike