已合并
swjdk-11.0.30-ga update #457
swjdk-11.0.30-ga update #457
已合并
lc-compiler创建于 3月27日
lc-compiler
3月27日

swjdk-11.0.30-ga update

likedislike
Pull Request已成功合入, 合并人@openeuler-ci-bot
(感谢 lc-compiler 的贡献)
Llc-compiler
3月27日 创建了 pull request,commit 4bfae499
openeuler-ci-botopeneuler-ci-bot成员
3月27日 将alexanderbill,li-yancheng,cf-zhao,eastb233,wangqiang95,kuenking111_admin,SegFault,chenzheng1030设为审查人
openeuler-ci-botopeneuler-ci-bot成员
3月27日 添加了label:sig/Compiler
openeuler-ci-bot
openeuler-ci-bot成员
3月27日 评论:

Welcome To openEuler Community

Hey @little-huaduo , thanks for your contribution to the community.

Bot Usage Manual

I'm the Bot here serving you. You can find the instructions on how to interact with me at Here . That means you can comment below every pull request or issue to trigger Bot Commands.

Contact Guide

If you have any questions, please contact the SIG: Compiler ,
and any of the maintainers: @SegFault, @alexanderbill, @cf-zhao, @chenzheng1030, @eastb233, @kuenking111_admin, @li-yancheng, @wangqiang95 ,
and any of the committers: @DXwangg, @dongxfeng, @douyiwang, @jiangfeilong1003, @xiezhaokun2013 .

likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月27日 添加了label:openeuler-cla/yes
openeuler-ci-bot
openeuler-ci-bot成员
3月27日 评论:

CLA Signature Pass

little-huaduo, thanks for your pull request. All authors of the commits have signed the CLA. 👍

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月27日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/openjdk-11/22/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月27日 添加了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
3月27日 评论:

当前仓库存在以下 保护分支 :

Protected Branch Version Release
master 11.0.30.7 4
openEuler-24.03-LTS-SP2 11.0.30.7 4
openEuler-24.03-LTS-SP1 11.0.30.7 4
openEuler-24.03-LTS-Next 11.0.30.7 4
openEuler-24.03-LTS 11.0.30.7 4
openEuler-22.03-LTS-SP4 11.0.30.7 4
openEuler-24.03-LTS-SP3 11.0.30.7 4
openEuler-25.09 11.0.28.6 1
openEuler-25.03 11.0.28.6 1
openEuler-20.03-LTS-SP4 11.0.28.6 1
sync-pr385-master-11.0.26-ga-riscv-to-openEuler-22.03-LTS-SP3 11.0.26.4 1
riscv64 11.0.11.9 9
openEuler-22.03-LTS-LoongArch 11.0.15.10 2

评论 /sync <branch1> <branch2> ... 可将当前 PR 修改同步到其它分支(创建同步 PR):
a) 如果当前 PR 是 Open 状态,同步操作将延迟到 PR 被合并时执行
b) 如果当前 PR 已经 Merged,将立即执行同步操作

注意:

  1. /sync 命令可以指定同步到多个分支,仅最后一个 /sync 命令生效
  2. 如果创建的同步 PR 不正确,可通过向同步 PR 的源分支提交轻量级 PR 完善,或使用 /close 命令关闭
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月27日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/openjdk-11/22/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月27日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/openjdk-11/22/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月27日 删除了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
3月27日 评论:
Check Name Build Result Build Details
check_binary_file ⚠WARNING #22
check_package_yaml_file ✅SUCCESS
check_repo_in_maintain ✅SUCCESS
check_consistency ✅SUCCESS
check_spec_file ✅SUCCESS
x86_64 check_build ❌FAILED #22
aarch64 check_build ❌FAILED #22
likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月27日 添加了label:ci_failed
openeuler-ci-bot
openeuler-ci-bot成员
3月27日 评论:
likedislike
kuenking111成员
3月28日 评论:

/retest

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月28日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/openjdk-11/23/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月28日 删除了label:ci_failed
openeuler-ci-botopeneuler-ci-bot成员
3月28日 添加了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
3月28日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/openjdk-11/23/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月28日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/openjdk-11/23/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月28日 删除了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
3月28日 评论:
Check Name Build Result Build Details
check_binary_file ⚠WARNING #23
check_package_yaml_file ✅SUCCESS
check_repo_in_maintain ✅SUCCESS
check_consistency ✅SUCCESS
check_spec_file ✅SUCCESS
x86_64 check_build ❌FAILED #23
aarch64 check_build ❌FAILED #23
likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月28日 添加了label:ci_failed
openeuler-ci-bot
openeuler-ci-bot成员
3月28日 评论:
likedislike
kuenking111成员
3月31日 评论:

/lgtm
/approve

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月31日 评论:

Review Code Feedback

  • The label lgtm was added to this pull request. It means that kuenking111_admin reviewed the code changes. 👋
Tips
  • If this pull request is not merged while all conditions are met, comment /check-pr to try again. 😄
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
3月31日 评论:

Review Code Feedback

  • The label approved was added to this pull request. It means that kuenking111_admin reviewed the code changes. 👋
Tips
  • If this pull request is not merged while all conditions are met, comment /check-pr to try again. 😄
likedislike
openeuler-ci-botopeneuler-ci-bot成员
3月31日 添加了label:approvedlgtm
openeuler-ci-botopeneuler-ci-bot成员
3月31日 合入了pull request,合并节点 SHA:79cdb0aeff7c7a97f767399c3a25d809fdd82d42