已开启
Fix CVE-2025-12060 #20
bidiudiu创建于 2025年12月5日
Fix CVE-2025-12060 #20
已开启
bidiudiu创建于 2025年12月5日
从refs/pull/20/head合入到master
bidiudiu
2025年12月5日
  1. Fix CVE-2025-12060
    #ID4B6I
    经过分析,低版本同样受影响,受影响的代码在keras/utils/data_utils.py
likedislike
合并受阻
Bbidiudiu
2025年12月5日 关联了 src-openeuler/python-Keras Issue ID4B6I
openeuler-ci-botopeneuler-ci-bot成员
2025年12月5日 指派了 chendh6 参与评审
openeuler-ci-botopeneuler-ci-bot成员
2025年12月5日 指派了 bingo 参与评审
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

Hi bidiudiu, welcome to the openEuler Community.
I'm the Bot here serving you. You can find the instructions on how to interact with me at Here.
If you have any questions, please contact the SIG: A-Tune, and any of the maintainers: @chendh6 , @hubin95 , @yangyongguang , @Vchanger , @znzjugod , @pshysimon

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年12月5日 指派了 znzjugod 参与评审
openeuler-ci-botopeneuler-ci-bot成员
2025年12月5日 指派了 PshySimon 参与评审
openeuler-ci-botopeneuler-ci-bot成员
2025年12月5日 指派了 yangyongguang 参与评审
openeuler-ci-botopeneuler-ci-bot成员
2025年12月5日 指派了 陈臻 参与评审
AtomGit-Bot
2025年12月5日 评论:

当前仓库存在以下 保护分支 :

Protected Branch Version Release
* master 2.12.0 2
openEuler-22.03-LTS-SP3 2.10.0 1
openEuler-22.03-LTS-SP4 2.10.0 1
openEuler-24.03-LTS 2.12.0 2
openEuler-24.03-LTS-Next 2.12.0 2
openEuler-24.03-LTS-SP1 2.12.0 2
openEuler-24.03-LTS-SP2 2.12.0 2
openEuler-24.03-LTS-SP3 2.12.0 2
openEuler-25.03 2.12.0 1
openEuler-25.09 2.12.0 2

评论 /sync <branch1> <branch2> ... 可将当前 PR 修改同步到其它分支(创建同步 PR):
a) 如果当前 PR 是 Open 状态,同步操作将延迟到 PR 被合并时执行
b) 如果当前 PR 已经 Merged,将立即执行同步操作

注意:

  1. /sync 命令可以指定同步到多个分支,仅最后一个 /sync 命令生效
  2. 如果创建的同步 PR 不正确,可通过向同步 PR 的源分支提交轻量级 PR 完善,或使用 /close 命令关闭
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年12月5日 添加了 &nbsp; openeuler-cla/yes 标签
openeuler-ci-botopeneuler-ci-bot成员
2025年12月5日 添加了 &nbsp; sig/A-Tune 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/python-Keras/41/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年12月5日 添加了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/python-Keras/37/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/python-Keras/40/console

likedislike
Bbidiudiu
2025年12月5日 强制推送 了代码
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

/retest

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/python-Keras/42/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年12月5日 移除了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:
Check Name Build Result Build Details
check_binary_file :white_check_mark:SUCCESS #41
check_package_yaml_file :white_check_mark:SUCCESS
check_repo_in_maintain :white_check_mark:SUCCESS
check_consistency :white_check_mark:SUCCESS
check_spec_file :white_check_mark:SUCCESS
aarch64 check_build :white_check_mark:SUCCESS #37
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
x86_64 check_build :white_check_mark:SUCCESS #40
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

如下为接口变更检查结果,目标分支为master,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms :white_check_mark:SUCCESS #40
delete_rpms :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
compare_package(aarch64) add_rpms :white_check_mark:SUCCESS #37
delete_rpms :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2025年12月5日 添加了 &nbsp; ci_successful 标签
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

@bidiudiu

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 41进行误报标记,41表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 41 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 41取消
也可在评论后加上一段文字描述,但请另起一行

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/python-Keras/38/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/python-Keras/41/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:
Check Name Build Result Build Details
check_binary_file :white_check_mark:SUCCESS #42
check_package_yaml_file :white_check_mark:SUCCESS
check_repo_in_maintain :white_check_mark:SUCCESS
check_consistency :white_check_mark:SUCCESS
check_spec_file :white_check_mark:SUCCESS
aarch64 check_build :white_check_mark:SUCCESS #38
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
x86_64 check_build :white_check_mark:SUCCESS #41
check_install :white_check_mark:SUCCESS
check_license :white_check_mark:SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

@bidiudiu

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

如下为接口变更检查结果,目标分支为master,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms :white_check_mark:SUCCESS #41
delete_rpms :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
compare_package(aarch64) add_rpms :white_check_mark:SUCCESS #38
delete_rpms :white_check_mark:SUCCESS
rpm_files :white_check_mark:SUCCESS
rpm_provides :white_check_mark:SUCCESS
rpm_requires :white_check_mark:SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2025年12月5日 评论:

门禁常见失败问题及解决方案, 可参考门禁问题排查手册
若门禁存在误报,您可以评论/ci_mistake 42进行误报标记,42表示本次构建号
也可带上误报的门禁检查项以及误报类型(ci、obs、infra),比如/ci_mistake 42 obs check_build check_install表示的是check_build和check_install存在误报,误报类型为obs
若想取消误报标记,可以评论/ci_unmistake 42取消
也可在评论后加上一段文字描述,但请另起一行

likedislike
bidiudiu
2025年12月8日 评论:

@chendh6 , @hubin95 , @yangyongguang , @Vchanger , @znzjugod , @pshysimon 辛苦各位老师审核

likedislike
bidiudiu
4月3日 评论:

@znzju @hubin95 辛苦各位老师有空审核一下

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
8月22日 评论:

/check-cla

likedislike
openeuler-ci-botopeneuler-ci-bot成员
8月22日 删除了label:openeuler-cla/yes
openeuler-ci-botopeneuler-ci-bot成员
8月22日 添加了label:openeuler-cla/yes
openeuler-ci-bot
openeuler-ci-bot成员
8月22日 评论:

CLA Signature Pass

bidiudiu, thanks for your pull request. All authors of the commits have signed the CLA. 👍

likedislike