已合并
Automatically generate code patches with openeuler !132 #298
AtomGit-Bot创建于 2021年6月2日
Automatically generate code patches with openeuler !132 #298
已合并
AtomGit-Bot创建于 2021年6月2日
refs/pull/298/head合入到openEuler-20.03-LTS-Next
AtomGit-Bot
2021年6月2日
spec: Update release version with !132

spec: Update patch and changelog with !132 fix Uninitialized Free Vulnerability  !132
bugfix: fix Uninitialized Free Vulnerability
likedislike
Pull Request已成功合入, 合并人@
(感谢 AtomGit-Bot 的贡献)
openeuler-ci-bot
openeuler-ci-bot成员
2021年6月2日 评论:

Hey kuhnchen18, Welcome to openEuler Community.
All of the projects in openEuler Community are maintained by @openeuler-ci-bot.
That means the developers can comment below every pull request or issue to trigger Bot Commands.
Please follow instructions at https://gitee.com/openeuler/community/blob/master/en/sig-infrastructure/command.md to find the details.

likedislike
AAtomGit-Bot
2021年6月2日 添加了   openeuler-cla/yes 标签
openeuler-ci-botopeneuler-ci-bot成员
2021年6月2日 添加了   stat/need-squash 标签
AAtomGit-Bot
2021年6月2日 移除了   stat/need-squash 标签
AAtomGit-Bot
2021年6月2日 移除了   openeuler-cla/yes 标签
AtomGit-Bot
2021年6月2日 评论:

当前仓库存在以下 保护分支

Protected Branch Version Release
master 4.1.0 58
openEuler-20.03-LTS 4.1.0 39
* openEuler-20.03-LTS-Next 4.1.0 46
openEuler-20.03-LTS-SP1 4.1.0 45
openEuler-20.03-LTS-SP2 4.1.0 47
openEuler-20.09 4.1.0 43
openEuler-21.03 4.1.0 55
riscv 5.2.0 1

评论 /sync <branch1> <branch2> ... 可将当前 PR 修改同步到其它分支(创建同步 PR):
a) 如果当前 PR 是 Open 状态,同步操作将延迟到 PR 被合并时执行
b) 如果当前 PR 已经 Merged,将立即执行同步操作

注意:

  1. /sync 命令可以指定同步到多个分支,仅最后一个 /sync 命令生效
  2. 如果创建的同步 PR 不正确,可通过向同步 PR 的源分支提交轻量级 PR 完善,或使用 /close 命令关闭
likedislike
AtomGit-Bot
2021年6月2日 评论:

@kuhnchen18, Thanks for your pull request. All the authors of commits have finished signinig CLA successfully. 👋

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2021年6月2日 添加了 &nbsp; stat/need-squash 标签
openeuler-ci-botopeneuler-ci-bot成员
2021年6月2日 添加了 &nbsp; sig/Virt 标签
openeuler-ci-botopeneuler-ci-bot成员
2021年6月2日 移除了 &nbsp; stat/need-squash 标签
imxcc成员
2021年6月2日 评论:

/retest

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2021年6月2日 添加了 &nbsp; ci_processing 标签
openeuler-ci-botopeneuler-ci-bot成员
2021年6月2日 添加了 &nbsp; stat/need-squash 标签
openeuler-ci-botopeneuler-ci-bot成员
2021年6月2日 移除了 &nbsp; ci_processing 标签
openeuler-ci-bot
openeuler-ci-bot成员
2021年6月2日 评论:
Check Name Build Result Build Details
check_code_style :white_check_mark:SUCCESS #280
check_package_license :white_check_mark:SUCCESS
check_package_yaml_file :white_check_mark:SUCCESS
check_spec_file :white_check_mark:SUCCESS
multiarch/src-openeuler/x86-64/qemu :white_check_mark:SUCCESS #276
multiarch/src-openeuler/aarch64/qemu :white_check_mark:SUCCESS #276
likedislike
openeuler-ci-botopeneuler-ci-bot成员
2021年6月2日 添加了 &nbsp; ci_successful 标签
openeuler-ci-bot
openeuler-ci-bot成员
2021年6月2日 评论:

@kuhnchen18

likedislike
imxcc成员
2021年6月2日 评论:

/check-cla
/check-pr
/lgtm
/approve

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2021年6月2日 添加了 &nbsp; lgtm 标签
openeuler-ci-bot
openeuler-ci-bot成员
2021年6月2日 评论:

lgtm is added in this pull request by: imxcc. 👋
NOTE:: If you find this pull request unmerged while all conditions meets, you are encouraged use command: "/check-pr" to try it again. 😄

likedislike
AAtomGit-Bot
2021年6月2日 添加了 &nbsp; openeuler-cla/yes 标签
AtomGit-Bot
2021年6月2日 评论:

@kuhnchen18, Thanks for your pull request. All the authors of commits have finished signinig CLA successfully. 👋

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
2021年6月2日 评论:

approved is added in this pull request by: imxcc. 👋
NOTE:: If you find this pull request unmerged while all conditions meets, you are encouraged use command: "/check-pr" to try it again. 😄

likedislike
openeuler-ci-botopeneuler-ci-bot成员
2021年6月2日 添加了 &nbsp; approved 标签
openeuler-ci-botopeneuler-ci-bot成员
2021年6月2日 合并了Pull Request(通过 API 合并)
openeuler-ci-bot
openeuler-ci-bot成员
2021年6月2日 评论:

The pull request merge failed, please use command "/check-pr" to try again.

likedislike