合并受阻
当前仓库存在以下 保护分支 :
| Protected Branch | Version | Release |
|---|---|---|
| master | 4.10.0.0 | 3 |
| openEuler-26.09-DevStation | 4.10.0.0 | 3 |
| openEuler-26.09 | 4.10.0.0 | 2 |
| openEuler-24.03-LTS-SP4 | 3.1.1 | 8 |
| openEuler-24.03-LTS-SP3 | 3.1.1 | 8 |
| openEuler-24.03-LTS-SP1 | 3.1.1 | 8 |
| openEuler-24.03-LTS-Next | 3.1.1 | 8 |
| openEuler-22.03-LTS-SP4 | 3.1.1 | 7 |
| openEuler-20.03-LTS-SP4 | 3.0.1 | 3 |
| add_pakage | 2.0.10 | 1 |
评论 /sync <branch1> <branch2> ... 可将当前 PR 修改同步到其它分支(创建同步 PR):
a) 如果当前 PR 是 Open 状态,同步操作将延迟到 PR 被合并时执行
b) 如果当前 PR 已经 Merged,将立即执行同步操作
注意:
- /sync 命令可以指定同步到多个分支,仅最后一个 /sync 命令生效
- 如果创建的同步 PR 不正确,可通过向同步 PR 的源分支提交轻量级 PR 完善,或使用 /close 命令关闭


门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/ukui-screensaver/9/console


Welcome To openEuler Community
Hey @fundawang , thanks for your contribution to the community.
Bot Usage Manual
I'm the Bot here serving you. You can find the instructions on how to interact with me at Here . That means you can comment below every pull request or issue to trigger Bot Commands. You can self-configure the PR merge rules for this repository. For more details, please refer to Here.
Contact Guide
If you have any questions, please contact the SIG: sig-UKUI ,
and any of the maintainers: @dou33, @hou-hongxun, @hua_yadong, @peijiankang ,
and any of the committers: @iaom1 .


aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/ukui-screensaver/9/console


x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/ukui-screensaver/9/console


/retest


门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/ukui-screensaver/10/console


x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/ukui-screensaver/10/console


aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/ukui-screensaver/10/console


当前门禁显示失败,是因为门禁的 RPM 存档丢失,不是真的构建失败。


当前PR是否有AI参与:
[x] 否
[ ] Yes
__1. AI Agent 平台:
__2. AI 模型 :
__3. Prompt 上下文 :
参考:openEuler社区《生成式AI工具使用与开源贡献政策》
PR功能描述 / 为什么需要这个合入**:
该PR关联的issue
(格式为fixes #<issue号>, 或者resolves #<issue号>): fixes #
希望检视人员了解: