已开启
update to version 1.64.04 #47
Funda Wang创建于 6 天前
update to version 1.64.04 #47
已开启
Funda Wang创建于 6 天前
Funda Wang
Funda Wang
6 天前

当前PR是否有AI参与:

[x] 否

[ ] Yes
__1. AI Agent 平台:
__2. AI 模型 :
__3. Prompt 上下文 :

参考:openEuler社区《生成式AI工具使用与开源贡献政策》

PR功能描述 / 为什么需要这个合入**:

该PR关联的issue

(格式为fixes #<issue号>, 或者resolves #<issue号>): fixes #

希望检视人员了解:

https://xmlrpc-c.sourceforge.io/change.html
由于自带的 expat 漏洞频发,xmlrpc-c 的下一版本 1.65 起默认使用系统 libxml2 编译。应跟随上游决策对上游依赖关系进行牵引。另外,由于不再使用 bundled expat,MIT 许可证自动消除。

likedislike
合并受阻
Funda WangFunda Wang
6 天前 创建了 pull request,commit 16c9196e
openeuler-ci-bot
openeuler-ci-bot成员
6 天前 评论:

当前仓库存在以下 保护分支 :

Protected Branch Version Release
master 1.64.03 3
openEuler-26.09 1.64.03 3
openEuler-24.03-LTS-Next 1.59.02 4
openEuler-26.09-DevStation 1.64.03 3
openEuler-24.03-LTS-SP4 1.59.02 4
openEuler-24.03-LTS-SP3 1.59.02 4
openEuler-24.03-LTS-SP1 1.59.02 4
openEuler-22.03-LTS-SP4 1.51.08 5
openEuler-20.03-LTS-SP4 1.51.06 4
openEuler1.0-base 1.51.03 3
openEuler1.0 1.51.03 3

评论 /sync <branch1> <branch2> ... 可将当前 PR 修改同步到其它分支(创建同步 PR):
a) 如果当前 PR 是 Open 状态,同步操作将延迟到 PR 被合并时执行
b) 如果当前 PR 已经 Merged,将立即执行同步操作

注意:

  1. /sync 命令可以指定同步到多个分支,仅最后一个 /sync 命令生效
  2. 如果创建的同步 PR 不正确,可通过向同步 PR 的源分支提交轻量级 PR 完善,或使用 /close 命令关闭
likedislike
openeuler-ci-botopeneuler-ci-bot成员
6 天前 将MrRlu,sunsuwan,Apricity_HW,robert-xingwang设为审查人
openeuler-ci-botopeneuler-ci-bot成员
6 天前 添加了label:sig/Networking
openeuler-ci-bot
openeuler-ci-bot成员
6 天前 评论:

Welcome To openEuler Community

Hey @fundawang , thanks for your contribution to the community.

Bot Usage Manual

I'm the Bot here serving you. You can find the instructions on how to interact with me at Here . That means you can comment below every pull request or issue to trigger Bot Commands. You can self-configure the PR merge rules for this repository. For more details, please refer to Here.

Contact Guide

If you have any questions, please contact the SIG: Networking ,
and any of the maintainers: @Apricity_HW, @MrRlu, @robert-xingwang, @sunsuwan ,
and any of the committers: @jiangheng12138, @zhongxuan2 .

likedislike
openeuler-ci-botopeneuler-ci-bot成员
6 天前 添加了label:openeuler-cla/yes
openeuler-ci-bot
openeuler-ci-bot成员
6 天前 评论:

CLA Signature Pass

fundawang, thanks for your pull request. All authors of the commits have signed the CLA. 👍

likedislike
Funda WangFunda Wang
6 天前 修改了pull request 的描述
openeuler-bot
openeuler-bot成员
6 天前 评论:
Check Name Build Result 详情 Build Details
x86_64 check_build ❌FAILED #34
aarch64 check_build ❌FAILED #34
likedislike
openeuler-botopeneuler-bot成员
6 天前 添加了label:ci_failed
openeuler-bot
openeuler-bot成员
6 天前 评论:
likedislike
Funda Wang
Funda Wang
1 天前 评论:

/retest

likedislike
openeuler-bot
openeuler-bot成员
1 天前 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/xmlrpc-c/35

likedislike
openeuler-botopeneuler-bot成员
1 天前 删除了label:ci_failed
openeuler-botopeneuler-bot成员
1 天前 添加了label:ci_processing
openeuler-bot
openeuler-bot成员
1 天前 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/xmlrpc-c/35

likedislike
openeuler-bot
openeuler-bot成员
1 天前 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/xmlrpc-c/35

likedislike
openeuler-botopeneuler-bot成员
1 天前 删除了label:ci_processing
openeuler-bot
openeuler-bot成员
1 天前 评论:
Check Name Build Result 详情 Build Details
check_binary_file ✅SUCCESS #35
check_lfsconfig ☑EXCLUDE
check_package_yaml_file ✅SUCCESS
check_repo_in_maintain ✅SUCCESS
check_consistency ✅SUCCESS
check_spec_file ✅SUCCESS
x86_64 check_build ✅SUCCESS #35
check_install ✅SUCCESS
check_license ✅SUCCESS
aarch64 check_build ✅SUCCESS #35
check_install ✅SUCCESS
check_license ✅SUCCESS
likedislike
openeuler-botopeneuler-bot成员
1 天前 添加了label:ci_successful
openeuler-bot
openeuler-bot成员
1 天前 评论:

如下为接口变更检查结果,目标分支为master,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms ✅SUCCESS #35
delete_rpms ✅SUCCESS
rpm_abi ✅SUCCESS
rpm_cmd ✅SUCCESS
rpm_files xmlrpc-c-devel ❌FAILED
rpm_header ✅SUCCESS
rpm_lib xmlrpc-c ❌FAILED
rpm_provides xmlrpc-c
xmlrpc-c-devel
❌FAILED
rpm_requires xmlrpc-c
xmlrpc-c-devel
❌FAILED
compare_package(aarch64) add_rpms ✅SUCCESS #35
delete_rpms ✅SUCCESS
rpm_abi ✅SUCCESS
rpm_cmd ✅SUCCESS
rpm_files xmlrpc-c-devel ❌FAILED
rpm_header ✅SUCCESS
rpm_lib xmlrpc-c ❌FAILED
rpm_provides xmlrpc-c
xmlrpc-c-devel
❌FAILED
rpm_requires xmlrpc-c
xmlrpc-c-devel
❌FAILED
likedislike
openeuler-bot
openeuler-bot成员
1 天前 评论:
likedislike