已开启
fix(hive): backport CVE-2024-29869 for openEuler-24.03-LTS-SP4 #46
fix(hive): backport CVE-2024-29869 for openEuler-24.03-LTS-SP4 #46
已开启
pop-code123创建于 6月10日
pop-code123
6月10日

当前PR是否有AI参与:

[x] 否

[ ] Yes
__1. AI Agent 平台:
__2. AI 模型 :
__3. Prompt 上下文 :

PR功能描述 / 为什么需要这个合入**:

修复 CVE-2024-29869

该PR关联的issue

#27

希望检视人员了解:

参考:https://github.com/apache/hive/commit/20106e254527f7d71b2e34455c4322e14950c620

likedislike
合并受阻
Ppop-code123
6月10日 创建了 pull request,commit 4cfc04ec
Ppop-code123
6月10日 关联了issue:CVE-2024-29869
openeuler-ci-botopeneuler-ci-bot成员
6月10日 将yangzhao_kl,macchen1设为审查人
openeuler-ci-botopeneuler-ci-bot成员
6月10日 添加了label:sig/bigdata
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:

Welcome To openEuler Community

Hey @pop-code123 , thanks for your contribution to the community.

Bot Usage Manual

I'm the Bot here serving you. You can find the instructions on how to interact with me at Here . That means you can comment below every pull request or issue to trigger Bot Commands.

Contact Guide

If you have any questions, please contact the SIG: bigdata ,
and any of the maintainers: @macchen1, @yangzhao_kl ,

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:

当前仓库存在以下 保护分支

Protected Branch Version Release
master 3.1.3 5
openEuler-24.03-LTS-SP4 3.1.3 3
openEuler-24.03-LTS-SP3 3.1.3 3
openEuler-24.03-LTS-SP1 3.1.3 3
openEuler-24.03-LTS-Next 3.1.3 3
openEuler-22.03-LTS-SP4 3.1.3 3
openEuler-20.03-LTS-SP4 3.1.3 1

评论 /sync <branch1> <branch2> ... 可将当前 PR 修改同步到其它分支(创建同步 PR):
a) 如果当前 PR 是 Open 状态,同步操作将延迟到 PR 被合并时执行
b) 如果当前 PR 已经 Merged,将立即执行同步操作

注意:

  1. /sync 命令可以指定同步到多个分支,仅最后一个 /sync 命令生效
  2. 如果创建的同步 PR 不正确,可通过向同步 PR 的源分支提交轻量级 PR 完善,或使用 /close 命令关闭
likedislike
openeuler-ci-botopeneuler-ci-bot成员
6月10日 添加了label:openeuler-cla/yes
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:

CLA Signature Pass

pop-code123, thanks for your pull request. All authors of the commits have signed the CLA. 👍

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/hive/6/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
6月10日 添加了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/hive/6/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/hive/6/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
6月10日 删除了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:
Check Name Build Result Build Details
check_binary_file WARNING #6
check_package_yaml_file SUCCESS
check_repo_in_maintain SUCCESS
check_consistency SUCCESS
check_spec_file SUCCESS
x86_64 check_build SUCCESS #6
check_install FAILED
check_license SUCCESS
aarch64 check_build SUCCESS #6
check_install SUCCESS
check_license SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
6月10日 添加了label:ci_failed
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:
likedislike
pop-code123
6月10日 评论:

/retest

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:

门禁正在运行, 您可以通过以下链接查看实时门禁检查结果.
若您对门禁结果含义不清晰或者遇到问题不知如何解决,可参考门禁指导手册
门禁入口及编码规范检查: multiarch/src-openeuler/trigger/hive/8/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
6月10日 删除了label:ci_failed
openeuler-ci-botopeneuler-ci-bot成员
6月10日 添加了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:

x86_64架构构建及构建后检查:multiarch/src-openeuler/x86-64/hive/8/console

likedislike
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:

aarch64架构构建及构建后检查:multiarch/src-openeuler/aarch64/hive/8/console

likedislike
openeuler-ci-botopeneuler-ci-bot成员
6月10日 删除了label:ci_processing
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:
Check Name Build Result Build Details
check_binary_file WARNING #8
check_package_yaml_file SUCCESS
check_repo_in_maintain SUCCESS
check_consistency SUCCESS
check_spec_file SUCCESS
x86_64 check_build SUCCESS #8
check_install SUCCESS
check_license SUCCESS
aarch64 check_build SUCCESS #8
check_install SUCCESS
check_license SUCCESS
likedislike
openeuler-ci-botopeneuler-ci-bot成员
6月10日 添加了label:ci_successful
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:

如下为接口变更检查结果,目标分支为openEuler-24.03-LTS-SP4,请PR提交者check差异信息

Arch Name Check Items Rpm Name Check Result Build Details
compare_package(x86_64) add_rpms SUCCESS #8
delete_rpms SUCCESS
rpm_files SUCCESS
rpm_header SUCCESS
rpm_jabi SUCCESS
rpm_lib SUCCESS
rpm_provides SUCCESS
rpm_requires SUCCESS
compare_package(aarch64) add_rpms SUCCESS #8
delete_rpms SUCCESS
rpm_files SUCCESS
rpm_header SUCCESS
rpm_jabi SUCCESS
rpm_lib SUCCESS
rpm_provides SUCCESS
rpm_requires SUCCESS
likedislike
openeuler-ci-bot
openeuler-ci-bot成员
6月10日 评论:
likedislike
pop-code123
6月10日 评论:
likedislike